Conversation
Signed-off-by: Carl Schwan <carl@carlschwan.eu>
If we find an circle mapping and the current user doesn't map to any of the group or user mapping, then load the circles for an user (and cache them). Signed-off-by: Carl Schwan <carl@carlschwan.eu>
perf: Optimize userInMappings with circle enabled
Signed-off-by: Nextcloud bot <bot@nextcloud.com>
Signed-off-by: Nextcloud bot <bot@nextcloud.com>
Signed-off-by: kristian.zendato <kristian.zendato@nextcloud.com>
Signed-off-by: Nextcloud bot <bot@nextcloud.com>
Signed-off-by: GitHub <noreply@github.com>
Signed-off-by: Louis <louis@chmn.me>
Signed-off-by: Nextcloud bot <bot@nextcloud.com>
fix: Update CODEOWNERS
Signed-off-by: Nextcloud bot <bot@nextcloud.com>
Signed-off-by: Carl Schwan <carl@carlschwan.eu>
And remove old stubs Signed-off-by: Carl Schwan <carl@carlschwan.eu>
Signed-off-by: Carl Schwan <carl@carlschwan.eu>
Signed-off-by: Carl Schwan <carl@carlschwan.eu>
…ck-in-acl-management Use strict comparison for circle ID matching in userInMappings
…-audit [master] Fix npm audit
Signed-off-by: Nextcloud bot <bot@nextcloud.com>
Signed-off-by: Ferdinand Thiessen <opensource@fthiessen.de>
fix: update group edit button accessibility and label
Signed-off-by: Nextcloud bot <bot@nextcloud.com>
Signed-off-by: Nextcloud bot <bot@nextcloud.com>
Signed-off-by: Nextcloud bot <bot@nextcloud.com>
Signed-off-by: Nextcloud bot <bot@nextcloud.com>
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
feat: create an .system folder in the team-folder after creating a team. This hidden folder will hold team-owned app-data in the future.
Relates to nextcloud/circles#2637
my PR #5011 has check fails
🤖 AI (if applicable)