Skip to content

CODEOWNERS: add jakecorrenti - #143

Merged
slp merged 1 commit into
libkrun:mainfrom
slp:add-jake
Sep 15, 2026
Merged

slp merged 1 commit into
libkrun:mainfrom
slp:add-jake

Conversation

@slp

@slp slp commented Sep 14, 2026

Copy link
Copy Markdown
Collaborator

No description provided.

Signed-off-by: Sergio Lopez <slp@redhat.com>
@slp
slp merged commit 3945f21 into libkrun:main Sep 15, 2026
6 checks passed
@slp
slp deleted the add-jake branch September 15, 2026 08:02
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants